Skip to content

chore: add "engines" field - #13437

Open
gameroman wants to merge 2 commits into
nextauthjs:mainfrom
gameroman:add-engines-field
Open

gameroman wants to merge 2 commits into
nextauthjs:mainfrom
gameroman:add-engines-field

Conversation

@gameroman

@gameroman gameroman commented May 27, 2026

Copy link
Copy Markdown

This makes it easier for consumers to know what to expect as far as compatibility goes

Both human consumers and static analysis tools

@gameroman
gameroman requested a review from ThangHuuVu as a code owner May 27, 2026 18:32
@vercel

vercel Bot commented May 27, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
auth-docs Ready Ready Preview Aug 2, 2026 12:07pm
1 Skipped Deployment
Project Deployment Actions Updated (UTC)
next-auth-docs Ignored Ignored Preview Aug 2, 2026 12:07pm

Request Review

@vercel

vercel Bot commented May 27, 2026

Copy link
Copy Markdown

@gameroman is attempting to deploy a commit to the authjs Team on Vercel.

A member of the Team first needs to authorize it.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant